[release-0.26] Bump fluxcd/source-controller to latest release - #5554
eks-distro-pr-bot wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Hi @eks-distro-pr-bot. Thanks for your PR. I'm waiting for a aws member to verify that this patch is reasonable to test. If it is, they should reply with Tip We noticed you've done this a few times! Consider joining the org to skip this step and gain Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
5b546c9 to
55f9188
Compare
54ef811 to
eb67b8f
Compare
8b472f7 to
f6ed2c4
Compare
f6ed2c4 to
60672ba
Compare
81fce1f to
210e21f
Compare
9291bef to
d1abc99
Compare
68a0fd3 to
d8d1ed7
Compare
58d6f6e to
539af0a
Compare
This PR bumps fluxcd/source-controller to the latest Git revision.
Compare changes
Release notes
/hold
/area dependencies
By submitting this pull request, I confirm that you can use, modify, copy, and redistribute this contribution, under the terms of your choice.