Skip to content

Bump aws-observability/aws-otel-collector to latest release - #5587

Open
eks-distro-pr-bot wants to merge 1 commit into
aws:mainfrom
eks-distro-pr-bot:update-aws-observability-aws-otel-collector-main
Open

eks-distro-pr-bot wants to merge 1 commit into
aws:mainfrom
eks-distro-pr-bot:update-aws-observability-aws-otel-collector-main

Conversation

@eks-distro-pr-bot

@eks-distro-pr-bot eks-distro-pr-bot commented Jul 15, 2026 •

Copy link
Copy Markdown
Contributor

This PR bumps aws-observability/aws-otel-collector to the latest Git revision.

Compare changes
Release notes

/hold
/area dependencies
/sig curated-packages

By submitting this pull request, I confirm that you can use, modify, copy, and redistribute this contribution, under the terms of your choice.

@eks-distro-bot eks-distro-bot added area/dependencies do-not-merge/hold sig/curated-packages Issue or pull request related to EKS-A Curated Packages size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. needs-ok-to-test labels Jul 15, 2026
@eks-distro-bot

Copy link
Copy Markdown
Collaborator

Hi @eks-distro-pr-bot. Thanks for your PR.

I'm waiting for a aws member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Tip

We noticed you've done this a few times! Consider joining the org to skip this step and gain /lgtm and other bot rights. We recommend asking approvers on your previous PRs to sponsor you.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-aws-observability-aws-otel-collector-main branch 5 times, most recently from 1ae5268 to af912ee Compare July 20, 2026 08:06
@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-aws-observability-aws-otel-collector-main branch from af912ee to 9fbc01c Compare September 1, 2026 08:05
@eks-distro-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign snarkychef for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

1 similar comment
@eks-distro-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign snarkychef for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-aws-observability-aws-otel-collector-main branch 7 times, most recently from abf8245 to e762ed9 Compare September 8, 2026 08:08
@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-aws-observability-aws-otel-collector-main branch 5 times, most recently from 7022ada to 2a047c6 Compare September 15, 2026 08:07
@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-aws-observability-aws-otel-collector-main branch 4 times, most recently from e1003ab to 4be3eb7 Compare September 19, 2026 08:07
@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-aws-observability-aws-otel-collector-main branch 8 times, most recently from 70da358 to f6761c2 Compare September 27, 2026 08:06
@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-aws-observability-aws-otel-collector-main branch 2 times, most recently from 7c21a37 to 311a239 Compare September 29, 2026 08:05
@eks-distro-pr-bot
eks-distro-pr-bot force-pushed the update-aws-observability-aws-otel-collector-main branch from 311a239 to 49e01a0 Compare September 30, 2026 08:05
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

area/dependencies do-not-merge/hold needs-ok-to-test sig/curated-packages Issue or pull request related to EKS-A Curated Packages size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants