Skip to content

Pin microshift-bootc image(s) for openshift-4.21 4.21.30 - #12406

Merged
openshift-art-build-bot[bot] merged 1 commit into
openshift-4.21from
auto-pin-microshift-bootc-openshift-4.21-4.21.30
Aug 19, 2026
Merged

openshift-art-build-bot[bot] merged 1 commit into
openshift-4.21from
auto-pin-microshift-bootc-openshift-4.21-4.21.30

Conversation

@openshift-art-build-bot

Copy link
Copy Markdown

@openshift-art-build-bot
openshift-art-build-bot Bot merged commit 59d69f2 into openshift-4.21 Aug 19, 2026
1 check passed
@openshift-ci

openshift-ci Bot commented Aug 19, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign san7ket for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-art-build-bot
openshift-art-build-bot Bot deleted the auto-pin-microshift-bootc-openshift-4.21-4.21.30 branch August 19, 2026 18:14
@openshift-ci

openshift-ci Bot commented Aug 19, 2026

Copy link
Copy Markdown
Contributor

@openshift-art-build-bot[bot]: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/art-pre-commit-check 514b165 link false /test art-pre-commit-check

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants